Skip to content

docs(product): Update stale image and descriptions on N+1 Queries Issue page - #19292

Open
inventarSarah wants to merge 1 commit into
masterfrom
smi/chore/stale-image-18205
Open

docs(product): Update stale image and descriptions on N+1 Queries Issue page#19292
inventarSarah wants to merge 1 commit into
masterfrom
smi/chore/stale-image-18205

Conversation

@inventarSarah

@inventarSarah inventarSarah commented Sep 8, 2026

Copy link
Copy Markdown
Collaborator

DESCRIBE YOUR PR

See: https://sentry-docs-git-smi-chorestale-image-18205.sentry.dev/product/issues/issue-details/performance-issues/n-one-queries/

  • updates 2 stale screenshots (settings, span evidence)
  • updates span evidence description
  • removes 2 screenshots from the "Example" section (These screenshots were made specifically for the example described, and in order to recreate them, we would have to recreate the whole example)
  • updates descriptions in the "Example" section

Closes: #18205

IS YOUR CHANGE URGENT?

Help us prioritize incoming PRs by letting us know when the change needs to go live.
Select exactly one option. For deadlines, replace YYYY-MM-DD with the due date. You can update this information later by editing the PR description.

  • Urgent deadline (GA date, etc.): YYYY-MM-DD
  • Other deadline: YYYY-MM-DD
  • No deadline: Not urgent, can wait up to 1 week+

SLA

  • Teamwork makes the dream work, so please add a reviewer to your PRs.
  • Please give the docs team up to 1 week to review your PR unless you've supplied a deadline.

Thanks in advance for your help!

PRE-MERGE CHECKLIST

Make sure you've checked the following before merging your changes:

  • Checked Vercel preview for correctness, including links
  • PR was reviewed and approved by any necessary SMEs (subject matter experts)
  • PR was reviewed and approved by a member of the Sentry docs team

EXTRA RESOURCES

@vercel

vercel Bot commented Sep 8, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Actions Updated
sentry-docs Ready Ready Preview Sep 8, 2026 8:43am UTC
1 Skipped Deployment
Project Deployment Actions Updated
develop-docs Ignored Ignored Sep 8, 2026 8:43am UTC

Request Review

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Priority: Normal Docs review has no urgent deadline

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Update stale images: product/issues/issue-details/performance-issues/n-one-queries (3 screenshots)

1 participant